Skip to content

Bump coreos/etcd from v3.6.13 to v3.7.1 - #253

Closed
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/docker/coreos/etcd-v3.7.0
Closed

Bump coreos/etcd from v3.6.13 to v3.7.1#253
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/docker/coreos/etcd-v3.7.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jul 8, 2026

Copy link
Copy Markdown
Contributor

Bumps coreos/etcd from v3.6.13 to v3.7.1.

@dependabot dependabot Bot added the kind/enhancement Enhancement, improvement, extension label Jul 8, 2026
@dependabot
dependabot Bot requested a review from a team as a code owner July 8, 2026 19:23
@gardener-prow

gardener-prow Bot commented Jul 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign vpnachev for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added cla: yes Indicates the PR's author has signed the cla-assistant.io CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 8, 2026
@dependabot
dependabot Bot force-pushed the dependabot/docker/coreos/etcd-v3.7.0 branch from 47522a6 to 3dd8801 Compare July 9, 2026 20:58
Bumps coreos/etcd from v3.6.13 to v3.7.1.

---
updated-dependencies:
- dependency-name: coreos/etcd
  dependency-version: v3.7.0
  dependency-type: direct:production
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot changed the title Bump coreos/etcd from v3.6.13 to v3.7.0 Bump coreos/etcd from v3.6.13 to v3.7.1 Jul 24, 2026
@dependabot
dependabot Bot force-pushed the dependabot/docker/coreos/etcd-v3.7.0 branch from 3dd8801 to 98860b6 Compare July 24, 2026 06:39
@gardener-prow gardener-prow Bot added cla: no Indicates the PR's author has not signed the cla-assistant.io CLA. cla: yes Indicates the PR's author has signed the cla-assistant.io CLA. and removed cla: yes Indicates the PR's author has signed the cla-assistant.io CLA. cla: no Indicates the PR's author has not signed the cla-assistant.io CLA. labels Jul 24, 2026
@dependabot @github

dependabot Bot commented on behalf of github Jul 24, 2026

Copy link
Copy Markdown
Contributor Author

Superseded by #256.

@dependabot dependabot Bot closed this Jul 24, 2026
@dependabot
dependabot Bot deleted the dependabot/docker/coreos/etcd-v3.7.0 branch July 24, 2026 20:09
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cla: yes Indicates the PR's author has signed the cla-assistant.io CLA. kind/enhancement Enhancement, improvement, extension size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants