Skip to content

Bump coreos/etcd from v3.6.13 to v3.7.1 - #256

Open
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/docker/coreos/etcd-v3.7.1
Open

Bump coreos/etcd from v3.6.13 to v3.7.1#256
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/docker/coreos/etcd-v3.7.1

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jul 24, 2026

Copy link
Copy Markdown
Contributor

Bumps coreos/etcd from v3.6.13 to v3.7.1.

Note
Automatic rebases have been disabled on this pull request as it has been open for over 30 days.

@dependabot dependabot Bot added the kind/enhancement Enhancement, improvement, extension label Jul 24, 2026
@dependabot
dependabot Bot requested a review from a team as a code owner July 24, 2026 20:09
@dependabot
dependabot Bot force-pushed the dependabot/docker/coreos/etcd-v3.7.1 branch from b707246 to 1136014 Compare July 24, 2026 20:09
@dependabot dependabot Bot added the kind/enhancement Enhancement, improvement, extension label Jul 24, 2026
@gardener-prow

gardener-prow Bot commented Jul 24, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign dimityrmirchev for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. cla: yes Indicates the PR's author has signed the cla-assistant.io CLA. labels Jul 24, 2026
Bumps coreos/etcd from v3.6.13 to v3.7.1.

---
updated-dependencies:
- dependency-name: coreos/etcd
  dependency-version: v3.7.1
  dependency-type: direct:production
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/docker/coreos/etcd-v3.7.1 branch from 1136014 to c996ee3 Compare August 20, 2026 07:03
@gardener-prow gardener-prow Bot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Aug 25, 2026
@gardener-prow

gardener-prow Bot commented Aug 25, 2026

Copy link
Copy Markdown

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cla: yes Indicates the PR's author has signed the cla-assistant.io CLA. kind/enhancement Enhancement, improvement, extension needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants